Update Konflux references - #774
Open
red-hat-konflux[bot] wants to merge 1 commit into
Open
red-hat-konflux[bot] wants to merge 1 commit into
red-hat-konflux[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Contributor
|
/retest |
3 similar comments
Contributor
|
/retest |
Contributor
|
/retest |
Contributor
|
/retest |
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
June 27, 2026 08:15
903d49c to
cc42777
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
July 4, 2026 08:09
cc42777 to
da57e2a
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
July 11, 2026 08:31
da57e2a to
b645e56
Compare
5 tasks
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
July 18, 2026 08:09
b645e56 to
2b7109a
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
July 25, 2026 08:22
2b7109a to
b9abbc6
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
2 times, most recently
from
August 8, 2026 08:18
45d8440 to
4ad0019
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
September 5, 2026 06:02
4ad0019 to
7da9727
Compare
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
red-hat-konflux
Bot
force-pushed
the
konflux/references/main
branch
from
September 19, 2026 00:34
7da9727 to
4f5439a
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
ccd3665→c177fc2290c9ec→1e2e00bd44fb2d→9eef3bfde35caf→8b50144c07d2be→fa53ef42e8fe30→a3678914be9343→c6c414a0.10.2→0.10.3ef00a86→3e59d639ef4dab→d9df5d347f4e2d→e92d00eafa8ba8→d00c88467a409d→17b958769d5fca→15d654c1808485→bbad5bfWarning
Some dependencies could not be looked up. Check the warning logs for more information.
Release Notes
konflux-ci/container-build-catalog (quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies-oci-ta)
v0.10.3Configuration
📅 Schedule: (UTC)
* * * * 6)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.