Skip to content

[react] Enable atom chromes only in edit mode - #621

Merged
sc-nikolaoslazaridis merged 1 commit into
release/betas/atomsfrom
feature/JSS-10208-2
Sep 9, 2026
Merged

sc-nikolaoslazaridis merged 1 commit into
release/betas/atomsfrom
feature/JSS-10208-2

Conversation

@sc-nikolaoslazaridis

Copy link
Copy Markdown
Contributor

Description / Motivation

Followup to the previous PR for JSS-10208 that enables the chromes only in edit mode

Testing Details

  • Unit Test Added
  • Manual Test/Other (Please elaborate)

Types of changes

  • Bug fix (non-breaking change which fixes an issue)
  • New feature (non-breaking change which adds functionality)
  • Breaking change (fix or feature that would cause existing functionality to change)

@changeset-bot

changeset-bot Bot commented Sep 9, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1dadb45

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sitecore-content-sdk/react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@sc-nikolaoslazaridis
sc-nikolaoslazaridis merged commit c876e45 into release/betas/atoms Sep 9, 2026
2 checks passed
@sc-nikolaoslazaridis
sc-nikolaoslazaridis deleted the feature/JSS-10208-2 branch September 9, 2026 15:13
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant